Skip to content

Update dependency pulpcore to v3.117.2 - #1476

Closed
red-hat-konflux[bot] wants to merge 1 commit into
mainfrom
konflux/mintmaker/main/pulpcore-3.x
Closed

red-hat-konflux[bot] wants to merge 1 commit into
mainfrom
konflux/mintmaker/main/pulpcore-3.x

Conversation

@red-hat-konflux

@red-hat-konflux red-hat-konflux Bot commented Sep 15, 2026

Copy link
Copy Markdown
Contributor

This PR contains the following updates:

Package Change Age Confidence
pulpcore (changelog) ==3.117.1==3.117.2 age confidence

Warning

Some dependencies could not be looked up. Check the warning logs for more information.


Release Notes

pulp/pulpcore (pulpcore)

v3.117.2

Compare Source

REST API {: #​3.117.2-rest-api }
Bugfixes {: #​3.117.2-rest-api-bugfix }
  • Preserve OpenPGP verification compatibility for failed signatures and fix key ID calculation for OpenPGP v6.
Plugin API {: #​3.117.2-plugin-api }

No significant changes.

Pulp File {: #​3.117.2-pulp-file }

No significant changes.

Pulp Cert Guard {: #​3.117.2-pulp-cert-guard }

No significant changes.



Configuration

📅 Schedule: (UTC)

  • Branch creation
    • At any time (no schedule defined)
  • Automerge
    • At any time (no schedule defined)

🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.

Rebasing: Whenever PR becomes conflicted, or you tick the rebase/retry checkbox.

🔕 Ignore: Close this PR and you won't be reminded about this update again.


  • If you want to rebase/retry this PR, check this box

To execute skipped test pipelines write comment /ok-to-test.


Documentation

Find out how to configure dependency updates in MintMaker documentation or see all available configuration options in Renovate documentation.

Signed-off-by: red-hat-konflux <126015336+red-hat-konflux[bot]@users.noreply.github.com>
@carlosthe19916

Copy link
Copy Markdown
Contributor

/retest

@red-hat-konflux red-hat-konflux Bot changed the title chore(deps): update dependency pulpcore to v3.117.2 Update dependency pulpcore to v3.117.2 Sep 17, 2026
@dkliban dkliban closed this Sep 18, 2026
@red-hat-konflux
red-hat-konflux Bot deleted the konflux/mintmaker/main/pulpcore-3.x branch September 19, 2026 03:12
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants