Skip to content

Sync from gdc-ui/sdk hash:3fbe577 - #7415

Merged
yenkins-admin merged 2 commits into
masterfrom
sync-from-gdc-ui-35046786688
Sep 16, 2026
Merged

yenkins-admin merged 2 commits into
masterfrom
sync-from-gdc-ui-35046786688

Conversation

@yenkins

@yenkins yenkins commented Sep 16, 2026

Copy link
Copy Markdown

Automated sync of gdc-ui → gooddata-ui-sdk Source hash: 3fbe577

@yenkins yenkins added the sync PR is created from gdc-ui label Sep 16, 2026

@yenkins-admin yenkins-admin left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Auto-approved by sync workflow

@coderabbitai

coderabbitai Bot commented Sep 16, 2026

Copy link
Copy Markdown

Important

Review skipped

Too many files!

This PR contains 415 files, which is 315 over the limit of 100.

To get a review, reduce the PR to 100 files or fewer by splitting it into smaller PRs or changing its base branch.

Upgrade to a paid plan to raise the limit.

Usage-priced reviews support at most 300 files.

⚙️ Run configuration

Configuration used: Organization UI

Review profile: CHILL

Plan: Advanced

Run ID: 3ef213b7-9bc9-48ff-873a-c63bd2b526d3

📥 Commits

Reviewing files that changed from the base of the PR and between a1ebfdf and e13874e.

⛔ Files ignored due to path filters (40)
  • common/autoinstallers/lint-staged/pnpm-lock.yaml is excluded by !**/pnpm-lock.yaml
  • common/config/rush/pnpm-lock.yaml is excluded by !**/pnpm-lock.yaml
  • libs/sdk-ui-tests-storybook/neobackstop/output/reference/storybook_chromium_10_Filters_DateFilter_-_full-featured_-_absolute-form_0_document_0_desktop.png is excluded by !**/*.png
  • libs/sdk-ui-tests-storybook/neobackstop/output/reference/storybook_chromium_10_Filters_DateFilter_-_granularity_switching_-_absolute-form-month_0_document_0_desktop.png is excluded by !**/*.png
  • libs/sdk-ui-tests-storybook/neobackstop/output/reference/storybook_chromium_10_Filters_DateFilter_-_granularity_switching_-_absolute-form-quarter_0_document_0_desktop.png is excluded by !**/*.png
  • libs/sdk-ui-tests-storybook/neobackstop/output/reference/storybook_chromium_10_Filters_DateFilter_-_granularity_switching_-_to-field-opens-after-from-field-click_0_document_0_desktop.png is excluded by !**/*.png
  • libs/sdk-ui-tests-storybook/neobackstop/output/reference/storybook_chromium_10_Filters_DateFilter_PeriodRangePicker_-_day_-_closed_0_document_0_desktop.png is excluded by !**/*.png
  • libs/sdk-ui-tests-storybook/neobackstop/output/reference/storybook_chromium_10_Filters_DateFilter_PeriodRangePicker_-_day_-_error_0_document_0_desktop.png is excluded by !**/*.png
  • libs/sdk-ui-tests-storybook/neobackstop/output/reference/storybook_chromium_10_Filters_DateFilter_PeriodRangePicker_-_day_-_opened-to_0_document_0_desktop.png is excluded by !**/*.png
  • libs/sdk-ui-tests-storybook/neobackstop/output/reference/storybook_chromium_10_Filters_DateFilter_PeriodRangePicker_-_day_-_opened_0_document_0_desktop.png is excluded by !**/*.png
  • libs/sdk-ui-tests-storybook/neobackstop/output/reference/storybook_chromium_10_Filters_DateFilter_PeriodRangePicker_-_day_themed_-_closed_0_document_0_desktop.png is excluded by !**/*.png
  • libs/sdk-ui-tests-storybook/neobackstop/output/reference/storybook_chromium_10_Filters_DateFilter_PeriodRangePicker_-_day_themed_-_error_0_document_0_desktop.png is excluded by !**/*.png
  • libs/sdk-ui-tests-storybook/neobackstop/output/reference/storybook_chromium_10_Filters_DateFilter_PeriodRangePicker_-_day_themed_-_opened-to_0_document_0_desktop.png is excluded by !**/*.png
  • libs/sdk-ui-tests-storybook/neobackstop/output/reference/storybook_chromium_10_Filters_DateFilter_PeriodRangePicker_-_day_themed_-_opened_0_document_0_desktop.png is excluded by !**/*.png
  • libs/sdk-ui-tests-storybook/neobackstop/output/reference/storybook_chromium_10_Filters_DateFilter_PeriodRangePicker_-_day_with_custom_format_-_closed_0_document_0_desktop.png is excluded by !**/*.png
  • libs/sdk-ui-tests-storybook/neobackstop/output/reference/storybook_chromium_10_Filters_DateFilter_PeriodRangePicker_-_day_with_custom_format_-_opened_0_document_0_desktop.png is excluded by !**/*.png
  • libs/sdk-ui-tests-storybook/neobackstop/output/reference/storybook_chromium_10_Filters_DateFilter_PeriodRangePicker_-_day_with_custom_hint_-_closed_0_document_0_desktop.png is excluded by !**/*.png
  • libs/sdk-ui-tests-storybook/neobackstop/output/reference/storybook_chromium_10_Filters_DateFilter_PeriodRangePicker_-_day_with_custom_hint_-_opened_0_document_0_desktop.png is excluded by !**/*.png
  • libs/sdk-ui-tests-storybook/neobackstop/output/reference/storybook_chromium_10_Filters_DateFilter_PeriodRangePicker_-_month_-_closed_0_document_0_desktop.png is excluded by !**/*.png
  • libs/sdk-ui-tests-storybook/neobackstop/output/reference/storybook_chromium_10_Filters_DateFilter_PeriodRangePicker_-_month_-_opened_0_document_0_desktop.png is excluded by !**/*.png
  • libs/sdk-ui-tests-storybook/neobackstop/output/reference/storybook_chromium_10_Filters_DateFilter_PeriodRangePicker_-_month_themed_-_closed_0_document_0_desktop.png is excluded by !**/*.png
  • libs/sdk-ui-tests-storybook/neobackstop/output/reference/storybook_chromium_10_Filters_DateFilter_PeriodRangePicker_-_month_themed_-_opened_0_document_0_desktop.png is excluded by !**/*.png
  • libs/sdk-ui-tests-storybook/neobackstop/output/reference/storybook_chromium_10_Filters_DateFilter_PeriodRangePicker_-_quarter_-_closed_0_document_0_desktop.png is excluded by !**/*.png
  • libs/sdk-ui-tests-storybook/neobackstop/output/reference/storybook_chromium_10_Filters_DateFilter_PeriodRangePicker_-_quarter_-_opened_0_document_0_desktop.png is excluded by !**/*.png
  • libs/sdk-ui-tests-storybook/neobackstop/output/reference/storybook_chromium_10_Filters_DateFilter_PeriodRangePicker_-_quarter_themed_-_closed_0_document_0_desktop.png is excluded by !**/*.png
  • libs/sdk-ui-tests-storybook/neobackstop/output/reference/storybook_chromium_10_Filters_DateFilter_PeriodRangePicker_-_quarter_themed_-_opened_0_document_0_desktop.png is excluded by !**/*.png
  • libs/sdk-ui-tests-storybook/neobackstop/output/reference/storybook_chromium_10_Filters_DateFilter_PeriodRangePicker_-_week_-_closed_0_document_0_desktop.png is excluded by !**/*.png
  • libs/sdk-ui-tests-storybook/neobackstop/output/reference/storybook_chromium_10_Filters_DateFilter_PeriodRangePicker_-_week_-_opened_0_document_0_desktop.png is excluded by !**/*.png
  • libs/sdk-ui-tests-storybook/neobackstop/output/reference/storybook_chromium_10_Filters_DateFilter_PeriodRangePicker_-_week_monday_start_-_closed_0_document_0_desktop.png is excluded by !**/*.png
  • libs/sdk-ui-tests-storybook/neobackstop/output/reference/storybook_chromium_10_Filters_DateFilter_PeriodRangePicker_-_week_monday_start_-_opened_0_document_0_desktop.png is excluded by !**/*.png
  • libs/sdk-ui-tests-storybook/neobackstop/output/reference/storybook_chromium_10_Filters_DateFilter_PeriodRangePicker_-_week_monday_start_themed_-_closed_0_document_0_desktop.png is excluded by !**/*.png
  • libs/sdk-ui-tests-storybook/neobackstop/output/reference/storybook_chromium_10_Filters_DateFilter_PeriodRangePicker_-_week_monday_start_themed_-_opened_0_document_0_desktop.png is excluded by !**/*.png
  • libs/sdk-ui-tests-storybook/neobackstop/output/reference/storybook_chromium_10_Filters_DateFilter_PeriodRangePicker_-_week_themed_-_closed_0_document_0_desktop.png is excluded by !**/*.png
  • libs/sdk-ui-tests-storybook/neobackstop/output/reference/storybook_chromium_10_Filters_DateFilter_PeriodRangePicker_-_week_themed_-_opened_0_document_0_desktop.png is excluded by !**/*.png
  • libs/sdk-ui-tests-storybook/neobackstop/output/reference/storybook_chromium_10_Filters_DateFilter_PeriodRangePicker_-_year_-_closed_0_document_0_desktop.png is excluded by !**/*.png
  • libs/sdk-ui-tests-storybook/neobackstop/output/reference/storybook_chromium_10_Filters_DateFilter_PeriodRangePicker_-_year_-_opened_0_document_0_desktop.png is excluded by !**/*.png
  • libs/sdk-ui-tests-storybook/neobackstop/output/reference/storybook_chromium_10_Filters_DateFilter_PeriodRangePicker_-_year_themed_-_closed_0_document_0_desktop.png is excluded by !**/*.png
  • libs/sdk-ui-tests-storybook/neobackstop/output/reference/storybook_chromium_10_Filters_DateFilter_PeriodRangePicker_-_year_themed_-_opened_0_document_0_desktop.png is excluded by !**/*.png
  • libs/sdk-ui-tests-storybook/neobackstop/output/reference/storybook_chromium_15_Ui_UiToolbarColorSwatch_-_default_0_document_0_desktop.png is excluded by !**/*.png
  • libs/sdk-ui-tests-storybook/neobackstop/output/reference/storybook_chromium_15_Ui_UiToolbarColorSwatch_-_themed_0_document_0_desktop.png is excluded by !**/*.png
📒 Files selected for processing (415)
  • NOTICE
  • common/autoinstallers/lint-staged/package.json
  • common/changes/@gooddata/sdk-ui-all/SHA_master_1_2026-09-15-15-20.json
  • common/changes/@gooddata/sdk-ui-all/SHA_master_1_2026-15-09-09-10.json
  • common/changes/@gooddata/sdk-ui-all/dkov-GDP-3376-step-4-manually-entering-date_2026-08-27-09-56-39.json
  • common/changes/@gooddata/sdk-ui-all/ine-richtext-computed-attribute_2026-09-14-15-39-11.json
  • common/changes/@gooddata/sdk-ui-all/jsc-f1-restricted-filters-entry_2026-09-14-20-50-00.json
  • common/changes/@gooddata/sdk-ui-all/mkn-olp-restricted-switcher-edit_2026-09-10-17-00-00.json
  • common/changes/@gooddata/sdk-ui-all/nce-f1-2737-alert-attribute-duplicate-menu-item_2026-09-09-11-12.json
  • common/changes/@gooddata/sdk-ui-all/pb-consolidate-postmessage_2026-09-14-14-51.json
  • common/changes/@gooddata/sdk-ui-all/pdo-lx-2941-reports-toolbar_2026-09-10-15-53-03.json
  • common/changes/@gooddata/sdk-ui-all/pdo-lx-3030-brand-kit-theme_2026-09-14-19-10-00.json
  • common/changes/@gooddata/sdk-ui-all/pvep-master-1_2026-09-15-11-34-16.json
  • common/changes/@gooddata/sdk-ui-all/pvep-master-2_2026-09-15-14-21-58.json
  • common/changes/@gooddata/sdk-ui-all/pvep-master-4_2026-09-14-13-18-13.json
  • common/changes/@gooddata/sdk-ui-all/pvep-master_2026-09-14-14-28-47.json
  • common/changes/@gooddata/sdk-ui-all/pvep-master_2026-09-15-15-11-42.json
  • common/changes/@gooddata/sdk-ui-all/rba-f1-2829-add-catalog-authoring-sl-cf-attributes-labels_2026-09-01-15-00-00.json
  • common/config/rush/version-policies.json
  • dev_docs/decisions.md
  • docs/content/en/latest/references/automations/_index.md
  • docs/content/en/latest/references/dashboard_component/_index.md
  • docs/content/en/latest/references/dashboard_component/automation_dialogs_customization.md
  • docs/content/en/latest/references/dashboard_component/detailed_description.md
  • docs/content/en/latest/references/embed_ai_assistant/_index.md
  • examples/playground/package.json
  • examples/sdk-interactive-examples/examples-template/package.json
  • examples/sdk-interactive-examples/examples/example-attributefilter/package.json
  • examples/sdk-interactive-examples/examples/example-chartconfig/package.json
  • examples/sdk-interactive-examples/examples/example-columnchart/package.json
  • examples/sdk-interactive-examples/examples/example-combochart/package.json
  • examples/sdk-interactive-examples/examples/example-dashboard/package.json
  • examples/sdk-interactive-examples/examples/example-datefilter/package.json
  • examples/sdk-interactive-examples/examples/example-dependentfilters/package.json
  • examples/sdk-interactive-examples/examples/example-execute/package.json
  • examples/sdk-interactive-examples/examples/example-granularity/package.json
  • examples/sdk-interactive-examples/examples/example-headline/package.json
  • examples/sdk-interactive-examples/examples/example-pivottable/package.json
  • examples/sdk-interactive-examples/examples/example-relativedatefilter/package.json
  • examples/sdk-interactive-examples/examples/example-repeater/package.json
  • examples/sdk-interactive-examples/package.json
  • libs/api-client-tiger/api/api-client-tiger.api.md
  • libs/api-client-tiger/package.json
  • libs/api-client-tiger/src/endpoints/entitiesObjects/index.ts
  • libs/api-client-tiger/src/metadataUtilities.ts
  • libs/sdk-backend-base/api/sdk-backend-base.api.md
  • libs/sdk-backend-base/package.json
  • libs/sdk-backend-base/src/decoratedBackend/attributes.ts
  • libs/sdk-backend-base/src/decoratedBackend/measures.ts
  • libs/sdk-backend-base/src/decoratedBackend/workspaceReports.ts
  • libs/sdk-backend-base/src/dummyBackend/InMemoryWorkspaceReportsService.test.ts
  • libs/sdk-backend-base/src/dummyBackend/InMemoryWorkspaceReportsService.ts
  • libs/sdk-backend-base/src/dummyBackend/index.ts
  • libs/sdk-backend-base/src/ldmFactories/metadata/attributeFactory.ts
  • libs/sdk-backend-base/src/ldmFactories/metadata/displayFormFactory.ts
  • libs/sdk-backend-mockingbird/package.json
  • libs/sdk-backend-mockingbird/src/recordedBackend/attributes.ts
  • libs/sdk-backend-mockingbird/src/recordedBackend/index.ts
  • libs/sdk-backend-spi/api/sdk-backend-spi.api.md
  • libs/sdk-backend-spi/package.json
  • libs/sdk-backend-spi/src/common/metadata.ts
  • libs/sdk-backend-spi/src/index.ts
  • libs/sdk-backend-spi/src/workspace/attributes/index.ts
  • libs/sdk-backend-spi/src/workspace/computedAttributes/index.ts
  • libs/sdk-backend-spi/src/workspace/dashboards/index.ts
  • libs/sdk-backend-spi/src/workspace/measures/index.ts
  • libs/sdk-backend-spi/src/workspace/reports/index.ts
  • libs/sdk-backend-tiger/package.json
  • libs/sdk-backend-tiger/src/backend/features/feature.ts
  • libs/sdk-backend-tiger/src/backend/settings/mapping.ts
  • libs/sdk-backend-tiger/src/backend/uiFeatures.ts
  • libs/sdk-backend-tiger/src/backend/workspace/attributes/elements/index.test.ts
  • libs/sdk-backend-tiger/src/backend/workspace/attributes/getAttributeDisplayForms.test.ts
  • libs/sdk-backend-tiger/src/backend/workspace/attributes/index.ts
  • libs/sdk-backend-tiger/src/backend/workspace/attributes/updateAttributeDisplayFormMeta.test.ts
  • libs/sdk-backend-tiger/src/backend/workspace/computedAttributes/index.ts
  • libs/sdk-backend-tiger/src/backend/workspace/dashboards/dashboardSideloads.test.ts
  • libs/sdk-backend-tiger/src/backend/workspace/dashboards/dashboardSideloads.ts
  • libs/sdk-backend-tiger/src/backend/workspace/dashboards/referenceAvailability.test.ts
  • libs/sdk-backend-tiger/src/backend/workspace/dashboards/referenceAvailability.ts
  • libs/sdk-backend-tiger/src/backend/workspace/measures/index.ts
  • libs/sdk-backend-tiger/src/backend/workspace/measures/updateMeasure.test.ts
  • libs/sdk-backend-tiger/src/backend/workspace/measures/updateMeasureMeta.test.ts
  • libs/sdk-backend-tiger/src/backend/workspace/reports.test.ts
  • libs/sdk-backend-tiger/src/backend/workspace/reports.ts
  • libs/sdk-backend-tiger/src/backend/workspace/settings/index.ts
  • libs/sdk-backend-tiger/src/convertors/fromBackend/MetadataConverter.test.ts
  • libs/sdk-backend-tiger/src/convertors/fromBackend/MetadataConverter.ts
  • libs/sdk-backend-tiger/src/convertors/fromBackend/MetricConverter.ts
  • libs/sdk-backend-tiger/src/convertors/fromBackend/conditionalFormattingConversions.ts
  • libs/sdk-backend-tiger/src/convertors/fromBackend/genAIConvertor.test.ts
  • libs/sdk-backend-tiger/src/convertors/fromBackend/genAIConvertor.ts
  • libs/sdk-backend-tiger/src/convertors/toBackend/ConditionalFormattingConverter.ts
  • libs/sdk-backend-tiger/src/utils/axiosResponse.test.helpers.ts
  • libs/sdk-code-convertors/api/sdk-code-convertors.api.md
  • libs/sdk-code-convertors/package.json
  • libs/sdk-code-convertors/python/pyproject.toml
  • libs/sdk-code-convertors/src/configs/table.ts
  • libs/sdk-code-schemas/package.json
  • libs/sdk-e2e-utils/package.json
  • libs/sdk-embedding/package.json
  • libs/sdk-model/api/sdk-model.api.md
  • libs/sdk-model/package.json
  • libs/sdk-model/src/conditionalFormatting/index.ts
  • libs/sdk-model/src/index.ts
  • libs/sdk-model/src/ldm/metadata/attribute/index.ts
  • libs/sdk-model/src/ldm/metadata/attributeDisplayForm/index.ts
  • libs/sdk-model/src/reports/brandKit.test.ts
  • libs/sdk-model/src/reports/brandKit.ts
  • libs/sdk-model/src/reports/builtinPageLayouts.ts
  • libs/sdk-model/src/reports/reports.test.ts
  • libs/sdk-model/src/reports/slot.ts
  • libs/sdk-model/src/reports/variables.ts
  • libs/sdk-model/src/settings/settings.ts
  • libs/sdk-model/src/theme/index.ts
  • libs/sdk-pluggable-application-model/package.json
  • libs/sdk-ui-all/package.json
  • libs/sdk-ui-application-header/package.json
  • libs/sdk-ui-catalog/api/sdk-ui-catalog.api.md
  • libs/sdk-ui-catalog/package.json
  • libs/sdk-ui-catalog/src/asCode/AsCodeDeleteDialog.test.tsx
  • libs/sdk-ui-catalog/src/catalogDetail/CatalogDetailAttributeLabelsConditionalFormatting.test.tsx
  • libs/sdk-ui-catalog/src/catalogDetail/CatalogDetailAttributeLabelsConditionalFormatting.tsx
  • libs/sdk-ui-catalog/src/catalogDetail/CatalogDetailConditionalFormatting.test.tsx
  • libs/sdk-ui-catalog/src/catalogDetail/CatalogDetailConditionalFormatting.tsx
  • libs/sdk-ui-catalog/src/catalogDetail/CatalogDetailContent.tsx
  • libs/sdk-ui-catalog/src/catalogDetail/CatalogDetailTabMetadata.test.tsx
  • libs/sdk-ui-catalog/src/catalogDetail/CatalogDetailTabMetadata.tsx
  • libs/sdk-ui-catalog/src/catalogDetail/hooks/useCatalogItemLoad.test.ts
  • libs/sdk-ui-catalog/src/catalogDetail/hooks/useCatalogItemLoad.ts
  • libs/sdk-ui-catalog/src/catalogDetail/hooks/useCatalogItemUpdate.ts
  • libs/sdk-ui-catalog/src/catalogDetail/useCatalogItemUpdate.test.ts
  • libs/sdk-ui-catalog/src/catalogItem/converter.test.ts
  • libs/sdk-ui-catalog/src/catalogItem/query.ts
  • libs/sdk-ui-catalog/src/catalogItem/testFixtures.ts
  • libs/sdk-ui-catalog/src/catalogItem/types.ts
  • libs/sdk-ui-catalog/src/computedAttribute/computedAttributeDescriptor.ts
  • libs/sdk-ui-catalog/src/computedAttribute/computedAttributeMutationPort.test.ts
  • libs/sdk-ui-catalog/src/computedAttribute/computedAttributeMutationPort.ts
  • libs/sdk-ui-catalog/src/localization/bundles/de-DE.json
  • libs/sdk-ui-catalog/src/localization/bundles/en-AU.json
  • libs/sdk-ui-catalog/src/localization/bundles/en-GB.json
  • libs/sdk-ui-catalog/src/localization/bundles/en-US.json
  • libs/sdk-ui-catalog/src/localization/bundles/es-419.json
  • libs/sdk-ui-catalog/src/localization/bundles/es-ES.json
  • libs/sdk-ui-catalog/src/localization/bundles/fi-FI.json
  • libs/sdk-ui-catalog/src/localization/bundles/fr-CA.json
  • libs/sdk-ui-catalog/src/localization/bundles/fr-FR.json
  • libs/sdk-ui-catalog/src/localization/bundles/id-ID.json
  • libs/sdk-ui-catalog/src/localization/bundles/it-IT.json
  • libs/sdk-ui-catalog/src/localization/bundles/ja-JP.json
  • libs/sdk-ui-catalog/src/localization/bundles/ko-KR.json
  • libs/sdk-ui-catalog/src/localization/bundles/nl-NL.json
  • libs/sdk-ui-catalog/src/localization/bundles/pl-PL.json
  • libs/sdk-ui-catalog/src/localization/bundles/pt-BR.json
  • libs/sdk-ui-catalog/src/localization/bundles/pt-PT.json
  • libs/sdk-ui-catalog/src/localization/bundles/ru-RU.json
  • libs/sdk-ui-catalog/src/localization/bundles/sl-SI.json
  • libs/sdk-ui-catalog/src/localization/bundles/th-TH.json
  • libs/sdk-ui-catalog/src/localization/bundles/tr-TR.json
  • libs/sdk-ui-catalog/src/localization/bundles/uk-UA.json
  • libs/sdk-ui-catalog/src/localization/bundles/vi-VN.json
  • libs/sdk-ui-catalog/src/localization/bundles/zh-HK.json
  • libs/sdk-ui-catalog/src/localization/bundles/zh-Hans.json
  • libs/sdk-ui-catalog/src/localization/bundles/zh-Hant.json
  • libs/sdk-ui-catalog/styles/scss/detail.scss
  • libs/sdk-ui-charts/package.json
  • libs/sdk-ui-charts/src/highcharts/chartTypes/_chartCreators/customTooltip/referenceResolver.test.ts
  • libs/sdk-ui-charts/src/highcharts/chartTypes/_chartCreators/customTooltip/referenceResolver.ts
  • libs/sdk-ui-charts/src/highcharts/chartTypes/_chartCreators/customTooltip/tooltipLookup.ts
  • libs/sdk-ui-dashboard/api/sdk-ui-dashboard.api.md
  • libs/sdk-ui-dashboard/package.json
  • libs/sdk-ui-dashboard/src/_staging/dateFilterConfig/merge.ts
  • libs/sdk-ui-dashboard/src/model/commandHandlers/dashboard/initializeDashboardHandler/loadUnavailableReferences.test.ts
  • libs/sdk-ui-dashboard/src/model/commandHandlers/dashboard/initializeDashboardHandler/loadUnavailableReferences.ts
  • libs/sdk-ui-dashboard/src/model/react/useIsWidgetRestricted.test.tsx
  • libs/sdk-ui-dashboard/src/model/react/useIsWidgetRestricted.ts
  • libs/sdk-ui-dashboard/src/model/store/catalog/catalogReducers.ts
  • libs/sdk-ui-dashboard/src/model/store/unavailableObjects/unavailableObjects.test.ts
  • libs/sdk-ui-dashboard/src/model/store/unavailableObjects/unavailableObjectsSelectors.ts
  • libs/sdk-ui-dashboard/src/presentation/automations/alerting/DefaultAlertingDialog/DefaultAlertingDialogAttribute.test.tsx
  • libs/sdk-ui-dashboard/src/presentation/automations/alerting/DefaultAlertingDialog/DefaultAlertingDialogAttribute.tsx
  • libs/sdk-ui-dashboard/src/presentation/automations/alerting/hooks/useAttributeValuesFromExecResults.ts
  • libs/sdk-ui-dashboard/src/presentation/automations/alerting/state/types.ts
  • libs/sdk-ui-dashboard/src/presentation/automations/alerting/types.ts
  • libs/sdk-ui-dashboard/src/presentation/automations/alerting/utils/getters.ts
  • libs/sdk-ui-dashboard/src/presentation/automations/alerting/utils/transform.test.ts
  • libs/sdk-ui-dashboard/src/presentation/filterBar/filterBar/RestrictedFiltersPlaceholder.test.tsx
  • libs/sdk-ui-dashboard/src/presentation/filterBar/filterBar/RestrictedFiltersPlaceholder.tsx
  • libs/sdk-ui-dashboard/src/presentation/flexibleLayout/dragAndDrop/Resize/HeightResizerHotspot.tsx
  • libs/sdk-ui-dashboard/src/presentation/localization/bundles/de-DE.json
  • libs/sdk-ui-dashboard/src/presentation/localization/bundles/en-AU.json
  • libs/sdk-ui-dashboard/src/presentation/localization/bundles/en-GB.json
  • libs/sdk-ui-dashboard/src/presentation/localization/bundles/en-US.json
  • libs/sdk-ui-dashboard/src/presentation/localization/bundles/es-419.json
  • libs/sdk-ui-dashboard/src/presentation/localization/bundles/es-ES.json
  • libs/sdk-ui-dashboard/src/presentation/localization/bundles/fi-FI.json
  • libs/sdk-ui-dashboard/src/presentation/localization/bundles/fr-CA.json
  • libs/sdk-ui-dashboard/src/presentation/localization/bundles/fr-FR.json
  • libs/sdk-ui-dashboard/src/presentation/localization/bundles/id-ID.json
  • libs/sdk-ui-dashboard/src/presentation/localization/bundles/it-IT.json
  • libs/sdk-ui-dashboard/src/presentation/localization/bundles/ja-JP.json
  • libs/sdk-ui-dashboard/src/presentation/localization/bundles/ko-KR.json
  • libs/sdk-ui-dashboard/src/presentation/localization/bundles/nl-NL.json
  • libs/sdk-ui-dashboard/src/presentation/localization/bundles/pl-PL.json
  • libs/sdk-ui-dashboard/src/presentation/localization/bundles/pt-BR.json
  • libs/sdk-ui-dashboard/src/presentation/localization/bundles/pt-PT.json
  • libs/sdk-ui-dashboard/src/presentation/localization/bundles/ru-RU.json
  • libs/sdk-ui-dashboard/src/presentation/localization/bundles/sl-SI.json
  • libs/sdk-ui-dashboard/src/presentation/localization/bundles/th-TH.json
  • libs/sdk-ui-dashboard/src/presentation/localization/bundles/tr-TR.json
  • libs/sdk-ui-dashboard/src/presentation/localization/bundles/uk-UA.json
  • libs/sdk-ui-dashboard/src/presentation/localization/bundles/vi-VN.json
  • libs/sdk-ui-dashboard/src/presentation/localization/bundles/zh-HK.json
  • libs/sdk-ui-dashboard/src/presentation/localization/bundles/zh-Hans.json
  • libs/sdk-ui-dashboard/src/presentation/localization/bundles/zh-Hant.json
  • libs/sdk-ui-dashboard/src/presentation/widget/common/RestrictedPlaceholder.test.tsx
  • libs/sdk-ui-dashboard/src/presentation/widget/common/RestrictedPlaceholder.tsx
  • libs/sdk-ui-dashboard/src/presentation/widget/common/configuration/DateDatasetDropdown.test.tsx
  • libs/sdk-ui-dashboard/src/presentation/widget/visualizationSwitcher/EditModeDashboardVisualizationSwitcher.test.tsx
  • libs/sdk-ui-dashboard/src/presentation/widget/visualizationSwitcher/EditModeDashboardVisualizationSwitcher.tsx
  • libs/sdk-ui-dashboard/src/presentation/widget/visualizationSwitcher/ViewModeDashboardVisualizationSwitcherRestricted.test.tsx
  • libs/sdk-ui-dashboard/src/presentation/widget/visualizationSwitcher/configuration/VisualizationsList/VisualizationListItem.restricted.test.tsx
  • libs/sdk-ui-dashboard/src/presentation/widget/visualizationSwitcher/configuration/VisualizationsList/VisualizationListItem.tsx
  • libs/sdk-ui-dashboard/styles/scss/dashboard.scss
  • libs/sdk-ui-dashboard/styles/scss/filterBar.scss
  • libs/sdk-ui-ext/api/sdk-ui-ext.api.md
  • libs/sdk-ui-ext/package.json
  • libs/sdk-ui-ext/src/insightPicker/InsightPickerCore.tsx
  • libs/sdk-ui-ext/src/internal/components/configurationControls/conditionalFormatting/ConditionalFormattingSection.tsx
  • libs/sdk-ui-ext/src/internal/translations/de-DE.json
  • libs/sdk-ui-ext/src/internal/translations/en-AU.json
  • libs/sdk-ui-ext/src/internal/translations/en-GB.json
  • libs/sdk-ui-ext/src/internal/translations/es-419.json
  • libs/sdk-ui-ext/src/internal/translations/es-ES.json
  • libs/sdk-ui-ext/src/internal/translations/fi-FI.json
  • libs/sdk-ui-ext/src/internal/translations/fr-CA.json
  • libs/sdk-ui-ext/src/internal/translations/fr-FR.json
  • libs/sdk-ui-ext/src/internal/translations/id-ID.json
  • libs/sdk-ui-ext/src/internal/translations/it-IT.json
  • libs/sdk-ui-ext/src/internal/translations/ja-JP.json
  • libs/sdk-ui-ext/src/internal/translations/ko-KR.json
  • libs/sdk-ui-ext/src/internal/translations/nl-NL.json
  • libs/sdk-ui-ext/src/internal/translations/pl-PL.json
  • libs/sdk-ui-ext/src/internal/translations/pt-BR.json
  • libs/sdk-ui-ext/src/internal/translations/pt-PT.json
  • libs/sdk-ui-ext/src/internal/translations/ru-RU.json
  • libs/sdk-ui-ext/src/internal/translations/sl-SI.json
  • libs/sdk-ui-ext/src/internal/translations/th-TH.json
  • libs/sdk-ui-ext/src/internal/translations/tr-TR.json
  • libs/sdk-ui-ext/src/internal/translations/uk-UA.json
  • libs/sdk-ui-ext/src/internal/translations/vi-VN.json
  • libs/sdk-ui-ext/src/internal/translations/zh-HK.json
  • libs/sdk-ui-ext/src/internal/translations/zh-Hans.json
  • libs/sdk-ui-ext/src/internal/translations/zh-Hant.json
  • libs/sdk-ui-ext/src/share/objectShareController.helpers.ts
  • libs/sdk-ui-filters/api/sdk-ui-filters.api.md
  • libs/sdk-ui-filters/package.json
  • libs/sdk-ui-filters/src/DateFilter/AbsoluteDateFilterForm/AbsoluteDateFilterForm.tsx
  • libs/sdk-ui-filters/src/DateFilter/AbsoluteDateFilterForm/conversions.test.ts
  • libs/sdk-ui-filters/src/DateFilter/AbsoluteDateFilterForm/conversions.ts
  • libs/sdk-ui-filters/src/DateFilter/DateFilterBody/DateFilterBody.test.tsx
  • libs/sdk-ui-filters/src/DateFilter/DateFilterBody/DateFilterBody.tsx
  • libs/sdk-ui-filters/src/DateFilter/DateFilterBody/DateFilterFormContent.tsx
  • libs/sdk-ui-filters/src/DateFilter/PeriodRangePicker/AccessibleFieldInput.tsx
  • libs/sdk-ui-filters/src/DateFilter/PeriodRangePicker/PeriodRangePickerImpl.tsx
  • libs/sdk-ui-filters/src/DateFilter/PeriodRangePicker/dateFnsRangePicker.tsx
  • libs/sdk-ui-filters/src/DateFilter/PeriodRangePicker/momentRangePicker.tsx
  • libs/sdk-ui-filters/src/DateFilter/PeriodRangePicker/periodRangePickerAccessibility.ts
  • libs/sdk-ui-filters/src/DateFilter/PeriodRangePicker/tests/PeriodRangePicker.test.tsx
  • libs/sdk-ui-filters/src/DateFilter/PeriodRangePicker/types.ts
  • libs/sdk-ui-filters/src/DateFilter/PeriodRangePicker/weekStartDateFnsLocale.ts
  • libs/sdk-ui-filters/src/DateFilter/PeriodRangePicker/weekStartMomentLocale.ts
  • libs/sdk-ui-filters/src/DateFilter/accessibility/keyboardNavigation.ts
  • libs/sdk-ui-filters/src/DateFilter/constants/Platform.ts
  • libs/sdk-ui-filters/src/DateFilter/utils/StaticPeriodConversions.ts
  • libs/sdk-ui-filters/src/FilterGroup/FilterGroup.tsx
  • libs/sdk-ui-filters/src/shared/utils/domUtilities.test.ts
  • libs/sdk-ui-filters/src/shared/utils/domUtilities.ts
  • libs/sdk-ui-filters/styles/scss/components/PeriodRangePicker.scss
  • libs/sdk-ui-gen-ai/api/sdk-ui-gen-ai.api.md
  • libs/sdk-ui-gen-ai/package.json
  • libs/sdk-ui-gen-ai/src/components/CustomizationContext.tsx
  • libs/sdk-ui-gen-ai/src/components/CustomizationProvider.tsx
  • libs/sdk-ui-gen-ai/src/components/GenAIChatConversations.tsx
  • libs/sdk-ui-gen-ai/src/components/customized/ConversationDateGrouping.tsx
  • libs/sdk-ui-gen-ai/src/components/customized/ConversationDrawerHeader.test.tsx
  • libs/sdk-ui-gen-ai/src/components/customized/ConversationDrawerHeader.tsx
  • libs/sdk-ui-gen-ai/src/components/customized/ConversationFooter.tsx
  • libs/sdk-ui-gen-ai/src/components/customized/ConversationHeader.tsx
  • libs/sdk-ui-gen-ai/src/components/customized/ConversationItem.tsx
  • libs/sdk-ui-gen-ai/src/components/customized/types.ts
  • libs/sdk-ui-gen-ai/src/components/intelligence/GenAiInteractionIntelligence.tsx
  • libs/sdk-ui-gen-ai/src/components/intelligence/InteractionIntelligenceTrigger.tsx
  • libs/sdk-ui-gen-ai/src/components/intelligence/detail/IntelligenceDetailRows.tsx
  • libs/sdk-ui-gen-ai/src/components/intelligence/detail/IntelligenceStepCarousel.tsx
  • libs/sdk-ui-gen-ai/src/components/intelligence/panel/IntelligenceCategoryRow.tsx
  • libs/sdk-ui-gen-ai/src/components/intelligence/panel/IntelligenceFooter.tsx
  • libs/sdk-ui-gen-ai/src/components/intelligence/panel/IntelligenceHeader.tsx
  • libs/sdk-ui-gen-ai/src/components/intelligence/panel/TimelineBar.tsx
  • libs/sdk-ui-gen-ai/src/components/messages/conversationContents/ConversationVisualisation.test.tsx
  • libs/sdk-ui-gen-ai/src/components/messages/conversationContents/ConversationVisualisation.tsx
  • libs/sdk-ui-gen-ai/src/index.ts
  • libs/sdk-ui-gen-ai/src/store/chatWindow/chatWindowSlice.ts
  • libs/sdk-ui-geo/package.json
  • libs/sdk-ui-geo/src/next/context/GeoLayersContext.test.tsx
  • libs/sdk-ui-geo/src/next/hooks/layers/useLayersTooltipData.test.tsx
  • libs/sdk-ui-geo/src/next/layers/common/customTooltipSection.test.ts
  • libs/sdk-ui-geo/src/next/layers/common/resolveReferencesFromGeoFeature.test.ts
  • libs/sdk-ui-geo/src/next/layers/common/resolveReferencesFromGeoFeature.ts
  • libs/sdk-ui-geo/src/next/layers/common/tooltipReferenceMaps.ts
  • libs/sdk-ui-geo/src/next/layers/registry/adapterTypes.ts
  • libs/sdk-ui-geo/src/next/layers/registry/output.test.ts
  • libs/sdk-ui-kit/api/sdk-ui-kit.api.md
  • libs/sdk-ui-kit/package.json
  • libs/sdk-ui-kit/src/@ui/UiColorPicker/UiColorPicker.test.tsx
  • libs/sdk-ui-kit/src/@ui/UiColorPicker/components/UiColorValueField.tsx
  • libs/sdk-ui-kit/src/@ui/UiToolbarColorSwatch/UiToolbarColorSwatch.scss
  • libs/sdk-ui-kit/src/@ui/UiToolbarColorSwatch/UiToolbarColorSwatch.test.tsx
  • libs/sdk-ui-kit/src/@ui/UiToolbarColorSwatch/UiToolbarColorSwatch.tsx
  • libs/sdk-ui-kit/src/@ui/UiToolbarStepper/UiToolbarStepper.scss
  • libs/sdk-ui-kit/src/@ui/UiToolbarStepper/UiToolbarStepper.test.tsx
  • libs/sdk-ui-kit/src/@ui/UiToolbarStepper/UiToolbarStepper.tsx
  • libs/sdk-ui-kit/src/RichText/RichText.tsx
  • libs/sdk-ui-kit/src/RichText/helpers/references.test.ts
  • libs/sdk-ui-kit/src/RichText/helpers/references.ts
  • libs/sdk-ui-kit/src/RichText/hooks/useEvaluatedMetricsAndAttributes.ts
  • libs/sdk-ui-kit/src/RichText/plugins/rehype.test.ts
  • libs/sdk-ui-kit/src/RichText/plugins/types.ts
  • libs/sdk-ui-loaders/package.json
  • libs/sdk-ui-pivot/package.json
  • libs/sdk-ui-pluggable-application/package.json
  • libs/sdk-ui-pluggable-host/package.json
  • libs/sdk-ui-semantic-search/package.json
  • libs/sdk-ui-tests-app/package.json
  • libs/sdk-ui-tests-e2e/package.json
  • libs/sdk-ui-tests-reference-workspace/package.json
  • libs/sdk-ui-tests-scenarios/package.json
  • libs/sdk-ui-tests-storybook/package.json
  • libs/sdk-ui-tests-storybook/stories/visual-regression/filters/DateFilter/DateFilter.stories.tsx
  • libs/sdk-ui-tests-storybook/stories/visual-regression/filters/DateFilter/PeriodRangePicker.stories.tsx
  • libs/sdk-ui-theme-provider/package.json
  • libs/sdk-ui-theme-provider/src/ThemeProvider/prepareTheme.test.ts
  • libs/sdk-ui-theme-provider/src/cssProperties.test.ts
  • libs/sdk-ui-theme-provider/src/cssProperties.ts
  • libs/sdk-ui-vis-commons/api/sdk-ui-vis-commons.api.md
  • libs/sdk-ui-vis-commons/package.json
  • libs/sdk-ui-vis-commons/src/customTooltip/referenceResolver.test.ts
  • libs/sdk-ui-vis-commons/src/customTooltip/referenceResolver.ts
  • libs/sdk-ui-vis-commons/src/customTooltip/tooltipExecution.test.ts
  • libs/sdk-ui-vis-commons/src/customTooltip/tooltipExecution.ts
  • libs/sdk-ui-vis-commons/src/customTooltip/tooltipLookup.ts
  • libs/sdk-ui-vis-commons/src/customTooltip/types.ts
  • libs/sdk-ui-vis-commons/src/customTooltip/useTooltipLookupExecutions.test.ts
  • libs/sdk-ui-vis-commons/src/index.ts
  • libs/sdk-ui-web-components/package.json
  • libs/sdk-ui/api/sdk-ui.api.md
  • libs/sdk-ui/package.json
  • libs/sdk-ui/src/base/localization/bundles/de-DE.json
  • libs/sdk-ui/src/base/localization/bundles/en-AU.json
  • libs/sdk-ui/src/base/localization/bundles/en-GB.json
  • libs/sdk-ui/src/base/localization/bundles/en-US.json
  • libs/sdk-ui/src/base/localization/bundles/es-419.json
  • libs/sdk-ui/src/base/localization/bundles/es-ES.json
  • libs/sdk-ui/src/base/localization/bundles/fi-FI.json
  • libs/sdk-ui/src/base/localization/bundles/fr-CA.json
  • libs/sdk-ui/src/base/localization/bundles/fr-FR.json
  • libs/sdk-ui/src/base/localization/bundles/id-ID.json
  • libs/sdk-ui/src/base/localization/bundles/it-IT.json
  • libs/sdk-ui/src/base/localization/bundles/ja-JP.json
  • libs/sdk-ui/src/base/localization/bundles/ko-KR.json
  • libs/sdk-ui/src/base/localization/bundles/nl-NL.json
  • libs/sdk-ui/src/base/localization/bundles/pl-PL.json
  • libs/sdk-ui/src/base/localization/bundles/pt-BR.json
  • libs/sdk-ui/src/base/localization/bundles/pt-PT.json
  • libs/sdk-ui/src/base/localization/bundles/ru-RU.json
  • libs/sdk-ui/src/base/localization/bundles/sl-SI.json
  • libs/sdk-ui/src/base/localization/bundles/th-TH.json
  • libs/sdk-ui/src/base/localization/bundles/tr-TR.json
  • libs/sdk-ui/src/base/localization/bundles/uk-UA.json
  • libs/sdk-ui/src/base/localization/bundles/vi-VN.json
  • libs/sdk-ui/src/base/localization/bundles/zh-HK.json
  • libs/sdk-ui/src/base/localization/bundles/zh-Hans.json
  • libs/sdk-ui/src/base/localization/bundles/zh-Hant.json
  • libs/sdk-ui/src/base/react/ErrorComponent.tsx
  • libs/util/package.json
  • templates/pluggable-module/harness/package.json
  • templates/pluggable-module/module/package.json
  • tools/app-toolkit/package.json
  • tools/catalog-export/package.json
  • tools/create-pluggable-module/package.json
  • tools/dashboard-plugin-template/package.json
  • tools/eslint-config/package.json
  • tools/eslint-config/src/configurations/react-hooks.ts
  • tools/eslint-config/src/configurations/react.ts
  • tools/eslint-config/src/index.ts
  • tools/eslint-config/src/types.ts
  • tools/i18n-toolkit/package.json
  • tools/lint-config/package.json
  • tools/lint-config/src/configurations/react-hooks.ts
  • tools/lint-config/src/configurations/react.ts
  • tools/lint-config/src/index.ts
  • tools/lint-config/src/types.ts
  • tools/mock-handling/package.json
  • tools/oxlint-config/PACKAGES.md
  • tools/oxlint-config/package.json
  • tools/oxlint-config/src/configurations/react-hooks.ts
  • tools/oxlint-config/src/configurations/react.ts
  • tools/oxlint-config/src/index.ts
  • tools/oxlint-config/src/types.ts
  • tools/plugin-toolkit/package.json
  • tools/react-app-template/package.json
  • tools/reference-workspace-mgmt/package.json
  • tools/reference-workspace/package.json
  • tools/stylelint-config/package.json

You can disable this status message by setting the reviews.review_status to false in the CodeRabbit configuration file.


Comment @coderabbitai help to get the list of available commands.

@yenkins-admin
yenkins-admin merged commit f5ab4c5 into master Sep 16, 2026
16 checks passed
@yenkins-admin
yenkins-admin deleted the sync-from-gdc-ui-35046786688 branch September 16, 2026 02:32
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

sync PR is created from gdc-ui

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants